Skip to content

Update version for a new release of the npm package. - #130

Merged
k8s-ci-robot merged 1 commit into
kubernetes-client:masterfrom
brendandburns:master
Oct 19, 2018
Merged

k8s-ci-robot merged 1 commit into
kubernetes-client:masterfrom
brendandburns:master

Conversation

@brendandburns

Copy link
Copy Markdown
Contributor

@itowlson @flyboarder

It's been a while since we pushed a release to NPM...

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: brendandburns

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot requested a review from mbohlool October 19, 2018 14:00
@k8s-ci-robot k8s-ci-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Oct 19, 2018
@burn2delete

Copy link
Copy Markdown
Contributor

Agreed! Most of the changes are highly desired lately.

@itowlson

Copy link
Copy Markdown
Contributor

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Oct 19, 2018
@k8s-ci-robot
k8s-ci-robot merged commit 6d012f0 into kubernetes-client:master Oct 19, 2018
@brendandburns

Copy link
Copy Markdown
Contributor Author

Version 0.7.0 pushed to npm

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants