feat(cli): Allow monorepos where all packages are fixed to use the root-style-tag (#1683) - #1796
Conversation
🦋 Changeset detectedLatest commit: 6a668be The changes in this PR will be included in the next version bump. Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
Codecov Report❌ Patch coverage is
Additional details and impacted files@@ Coverage Diff @@
## main #1796 +/- ##
==========================================
+ Coverage 84.24% 84.43% +0.18%
==========================================
Files 56 57 +1
Lines 2419 2428 +9
Branches 729 729
==========================================
+ Hits 2038 2050 +12
+ Misses 375 372 -3
Partials 6 6 ☔ View full report in Codecov by Harness. 🚀 New features to boost your workflow:
|
harrisonhoward
left a comment
There was a problem hiding this comment.
Minor changes needed but the end result is the same.
172d8ed to
260fad2
Compare
|
Thanks for this PR! This is the feature that we desperately need. |
…-short-hand-git-tag
|
Hey @beeequeue! Any chance that this one will be in v3 in future? |
Summary
Related issues
Closes #1638