Skip to content

Update CODEOWNERS to reference GitHub Enterprise team name - #33

Merged
lelia merged 1 commit into
mainfrom
lelia/add-codeowners-group
Feb 20, 2026
Merged

lelia merged 1 commit into
mainfrom
lelia/add-codeowners-group

Conversation

@lelia

@lelia lelia commented Feb 20, 2026

Copy link
Copy Markdown
Contributor

Now that we have a more narrowly-scoped CE team on GitHub Enterprise, we should map that to the CODEOWNERS file.

Signed-off-by: lelia <lelia@socket.dev>
@lelia
lelia merged commit 635f90f into main Feb 20, 2026
4 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant